Skip to content

[main] Upgrade to latest dependencies - #1153

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
knative-automation:auto-updates/update-deps-main
Sep 21, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
knative-automation:auto-updates/update-deps-main

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

Cron -dprotaso

/cc knative/serving-writers
/assign knative/serving-writers

Produced by: knative-extensions/knobots/actions/update-deps

bumping k8s.io/apimachinery 3e341b5...eb97e1c:
  > eb97e1c Merge pull request # 141225 from liggitt/golang-deps-1-35
  > b3ca310 Update golang.org/x deps
bumping k8s.io/api cfdecb5...b65ac35:
  > b65ac35 Update dependencies to v0.35.8 tag
  > 5066193 Merge pull request # 141225 from liggitt/golang-deps-1-35
  > 215d9ba Update golang.org/x deps
bumping google.golang.org/grpc 4c226da...030ee8b:
  > 030ee8b Update version to 1.83.2 (# 9375)
  > 8668b69 cherry-pick # 9365 to v1.83.x (# 9366)
  > a3e952d cherry-pick # 9346 to v1.83.x and update x/net dependency (# 9369)
  > 58f8fd9 Change version to 1.83.2-dev (# 9337)
  > 1550d9e Change version to 1.83.1 (# 9336)
  > ebba6f3 Cherry-pick # 9258 and # 9332 into v1.83.x (# 9335)
  > 8cfeca0 Cherry-pick # 9331 to v1.83.x (# 9333)
  > dec6951 Change version to 1.83.1-dev (# 9229)
bumping knative.dev/pkg c9015f8...5dc1978:
  > 5dc1978 Bump google.golang.org/grpc from 1.83.0 to 1.83.2 (# 3389)
  > d2a153a Bump the k8s group with 5 updates (# 3386)
bumping k8s.io/client-go d2a9618...3614e2d:
  > 3614e2d Update dependencies to v0.35.8 tag
  > 7bd3cb5 Merge pull request # 141225 from liggitt/golang-deps-1-35
  > 0fab4f6 Update golang.org/x deps
bumping k8s.io/code-generator 15ce9a0...1aaf60c:
  > 1aaf60c Update dependencies to v0.35.8 tag
  > e8766ba Merge pull request # 141225 from liggitt/golang-deps-1-35
  > 6a477bc Update golang.org/x deps

Signed-off-by: Knative Automation <automation@knative.team>
@knative-prow
knative-prow Bot requested a review from a team September 21, 2026 01:39
@knative-prow

knative-prow Bot commented Sep 21, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign dprotaso for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Sep 21, 2026
@codecov

codecov Bot commented Sep 21, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 93.34%. Comparing base (7d916c7) to head (36d6a35).
⚠️ Report is 1 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##             main    #1153   +/-   ##
=======================================
  Coverage   93.34%   93.34%           
=======================================
  Files          35       35           
  Lines        1007     1007           
=======================================
  Hits          940      940           
  Misses         53       53           
  Partials       14       14           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@knative-prow
knative-prow Bot merged commit e397784 into knative:main Sep 21, 2026
27 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/L Denotes a PR that changes 100-499 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant