Skip to content

docs(ruby): Update quick start guides - #19453

Open
inventarSarah wants to merge 3 commits into
masterfrom
smi/quick-start/ruby
Open

inventarSarah wants to merge 3 commits into
masterfrom
smi/quick-start/ruby

Conversation

@inventarSarah

Copy link
Copy Markdown
Collaborator

DESCRIBE YOUR PR

Updates the Ruby quick start guides to align with our updated template:

  • uses split layout
  • updates structure
  • adds learn more links to init snippets
  • adds a new section for data_collection
  • adds additional verification steps (tracing, logs, metrics)
  • removes "logs" onboarding option

Note

We're currently working on a better way to integrate always-enabled features (like logs and metrics) into the quick start guides (you're welcome to drop your feedback and ideas). For now, I've put logs and metrics under "Verify" but this will likely change!

IS YOUR CHANGE URGENT?

Help us prioritize incoming PRs by letting us know when the change needs to go live.
Select exactly one option. For deadlines, replace YYYY-MM-DD with the due date. You can update this information later by editing the PR description.

  • Urgent deadline (GA date, etc.): YYYY-MM-DD
  • Other deadline: YYYY-MM-DD
  • No deadline: Not urgent, can wait up to 1 week+

SLA

  • Teamwork makes the dream work, so please add a reviewer to your PRs.
  • Please give the docs team up to 1 week to review your PR unless you've supplied a deadline.

Thanks in advance for your help!

PRE-MERGE CHECKLIST

Make sure you've checked the following before merging your changes:

  • Checked Vercel preview for correctness, including links
  • PR was reviewed and approved by any necessary SMEs (subject matter experts)
  • PR was reviewed and approved by a member of the Sentry docs team

EXTRA RESOURCES

@vercel

vercel Bot commented Sep 17, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated
sentry-docs Ready Ready Preview Sep 17, 2026 8:34am UTC
1 Skipped Deployment
Project Deployment Actions Updated
develop-docs Ignored Ignored Sep 17, 2026 8:34am UTC

Request Review

@coolguyzone coolguyzone left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good! 👀

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Priority: Normal Docs review has no urgent deadline

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants